Skip to content

Create CODEOWNERS file with default owners - #65

Merged
martin-s-a merged 1 commit into
mainfrom
ms_codeowners
Oct 28, 2025
Merged

martin-s-a merged 1 commit into
mainfrom
ms_codeowners

Conversation

@martin-s-a

Copy link
Copy Markdown
Contributor

Add default code owners for the repository.
This will also allow dependabot to request review from code owners when opening a PR.

Add default code owners for the repository.
This will also allow dependabot to request review from code owners when opening a PR.
@martin-s-a martin-s-a self-assigned this Oct 27, 2025
@martin-s-a martin-s-a added the enhancement New feature or request label Oct 27, 2025
@sonarqubecloud

Copy link
Copy Markdown

@martin-s-a martin-s-a changed the title Add CODEOWNERS file for repository ownership Create CODEOWNERS file with default owners Oct 27, 2025
@martin-s-a
martin-s-a enabled auto-merge (squash) October 27, 2025 16:49
@martin-s-a
martin-s-a merged commit 61b4fab into main Oct 28, 2025
15 checks passed
@martin-s-a
martin-s-a deleted the ms_codeowners branch October 28, 2025 18:08
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

enhancement New feature or request

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants